Skip to content

Update CODEOWNERS - #4189

Merged
mesudip merged 3 commits into
IntersectMBO:developfrom
cardanoapi:chore/update-codeowners
Aug 26, 2026
Merged

Update CODEOWNERS#4189
mesudip merged 3 commits into
IntersectMBO:developfrom
cardanoapi:chore/update-codeowners

Conversation

@kusssal

@kusssal kusssal commented Aug 10, 2026

Copy link
Copy Markdown
Contributor

List of changes

  • Change: Update default and GitHub workflow code owners.

Checklist

  • My changes generate no new warnings
  • My code follows the style guidelines of this project
  • I have commented my code, particularly in hard-to-understand areas
  • I have made corresponding changes to the changelog
  • I have added tests that prove my fix is effective or that my feature works

@kusssal
kusssal requested a review from aaboyle878 as a code owner August 10, 2026 04:52
@mesudip
mesudip force-pushed the chore/update-codeowners branch from 5f4e8f5 to db3e249 Compare August 26, 2026 09:28
@mesudip
mesudip self-requested a review August 26, 2026 09:32
@mesudip
mesudip merged commit e018950 into IntersectMBO:develop Aug 26, 2026
3 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants