Skip to content

davis: 5.3.0 -> 5.4.1 - #498277

Closed
Ramblurr wants to merge 1 commit into
NixOS:masterfrom
Ramblurr:update/davis
Closed

Ramblurr wants to merge 1 commit into
NixOS:masterfrom
Ramblurr:update/davis

Conversation

@Ramblurr

@Ramblurr Ramblurr commented Mar 9, 2026

Copy link
Copy Markdown
Contributor

https://github.com/tchapi/davis/releases/tag/v5.4.0

Things done

  • Built on platform:
    • x86_64-linux
    • aarch64-linux
    • x86_64-darwin
    • aarch64-darwin
  • Tested, as applicable:
  • Ran nixpkgs-review on this PR. See nixpkgs-review usage.
  • Tested basic functionality of all binary files, usually in ./result/bin/.
  • Nixpkgs Release Notes
    • Package update: when the change is major or breaking.
  • NixOS Release Notes
    • Module addition: when adding a new NixOS module.
    • Module update: when the change is significant.
  • Fits CONTRIBUTING.md, pkgs/README.md, maintainers/README.md and other READMEs.

@nixpkgs-ci nixpkgs-ci Bot added 8.has: package (update) This PR updates a package to a newer version 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Mar 9, 2026
@Limosine

Limosine commented Mar 9, 2026

Copy link
Copy Markdown
Contributor

Works nicely on my server. nixpkgs-review also ran without errors.

nixpkgs-review result

Generated using nixpkgs-review.

Command: nixpkgs-review pr 498277
Commit: 09cd37fc306caea957028cd10b923d21c4dc048f


x86_64-linux

✅ 1 package built:
  • davis

@Ramblurr

Ramblurr commented Mar 10, 2026

Copy link
Copy Markdown
Contributor Author

@Limosine thanks for the feedback. Could you please leave a github review on the PR? It will appease the nixpkgs bot and help us get noticed by a nixpkgs committer sooner.

https://github.com/NixOS/nixpkgs/pull/498277/changes -> Submit Review [green button] -> Approve -> Submit Review

@Limosine Limosine left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the PR. LGTM.

@nixpkgs-ci nixpkgs-ci Bot added 12.approvals: 1 This PR was reviewed and approved by one person. and removed 9.needs: reviewer This PR currently has no reviewers requested and needs attention. labels Mar 10, 2026
@NickCao

NickCao commented Mar 11, 2026

Copy link
Copy Markdown
Member

What about the new API_KEY envvar, do we need to update the nixos module for it?

@Ramblurr

Ramblurr commented Mar 12, 2026

Copy link
Copy Markdown
Contributor Author

Hmm, nice catch. The release notes don't mention anything about an API_KEY.. can you clarify what it's used for and how it affects the nixos module?

@ameertaweel

Copy link
Copy Markdown
Contributor

I'm not sure if the release notes got updated after you checked, but they mention the API key:

Add (limited) API endpoint by unkn0wnAPI* in tchapi/davis#237

*: Removed @ before username to not mention them.

@ameertaweel

Copy link
Copy Markdown
Contributor

Also, v5.4.1 was released. Can you please update the PR to use it?

@ameertaweel

Copy link
Copy Markdown
Contributor

Regarding the API key, here are the docs for it:
https://github.com/tchapi/davis/blob/main/docs/api/README.md

As far as I understand, we can just pass it via services.davis.config. But then it would be visible in the Nix source. Maybe @NickCao was suggesting to add an option similar to services.davis.appSecretFile, like services.davis.apiKeyFile.

@Ramblurr

Copy link
Copy Markdown
Contributor Author

Also, v5.4.1 was released. Can you please update the PR to use it?

Incoming...

As far as I understand, we can just pass it via services.davis.config. But then it would be visible in the Nix source. Maybe @NickCao was suggesting to add an option similar to services.davis.appSecretFile, like services.davis.apiKeyFile.

This isn't strictly necessary, the module already supports passing arbitrary config values as files using _secret

services.davis.config = {
...
  API_KEY._secret = config.my.api-key.file.path;
...
};

@Ramblurr Ramblurr changed the title davis: 5.3.0 -> 5.4.0 davis: 5.3.0 -> 5.4.1 Mar 16, 2026
@ameertaweel

Copy link
Copy Markdown
Contributor

Thanks for updating the PR to v5.4.1 @Ramblurr.

Also, I totally missed the _secret thing. But now I don't see any needed changes in the NixOS module.

@NickCao can you clarify what you meant?

@NickCao

NickCao commented Mar 16, 2026

Copy link
Copy Markdown
Member

Thanks for updating the PR to v5.4.1 @Ramblurr.

Also, I totally missed the _secret thing. But now I don't see any needed changes in the NixOS module.

@NickCao can you clarify what you meant?

I'm just saying the upstream diff shows a new field API_KEY in the .env file, so it seems we don't have to do anything to support it. No change to the module is needed then.

@Ramblurr

Ramblurr commented Jun 1, 2026

Copy link
Copy Markdown
Contributor Author

I couldn't ever attract a nix committer to this PR :( but anyways this is done over in #526607 thanks to the bot

@NickCao

NickCao commented Jun 1, 2026

Copy link
Copy Markdown
Member

Superseded by #526607

@NickCao NickCao closed this Jun 1, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

8.has: package (update) This PR updates a package to a newer version 10.rebuild-darwin: 1-10 This PR causes between 1 and 10 packages to rebuild on Darwin. 10.rebuild-darwin: 1 This PR causes 1 package to rebuild on Darwin. 10.rebuild-linux: 1-10 This PR causes between 1 and 10 packages to rebuild on Linux. 10.rebuild-linux: 1 This PR causes 1 package to rebuild on Linux. 11.by: package-maintainer This PR was created by a maintainer of all the package it changes. 12.approvals: 1 This PR was reviewed and approved by one person.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants