Skip to content

go_1_14: 1.14 -> 1.14.1 - #83689

Merged
Mic92 merged 1 commit into
NixOS:stagingfrom
flokli:go-1.14.1
Apr 2, 2020
Merged

Mic92 merged 1 commit into
NixOS:stagingfrom
flokli:go-1.14.1

Conversation

@flokli

@flokli flokli commented Mar 29, 2020

Copy link
Copy Markdown
Member

This contains at least golang/go#37807, which
currently causes some of our go binaries to just fail when booted with
older kernels.

See golang/go#37436 for context.

Things done
  • Tested using sandboxing (nix.useSandbox on NixOS, or option sandbox in nix.conf on non-NixOS linux)
  • Built on platform(s)
    • NixOS
    • macOS
    • other Linux distributions
  • Tested via one or more NixOS test(s) if existing and applicable for the change (look inside nixos/tests)
  • Tested compilation of all pkgs that depend on this change using nix-shell -p nixpkgs-review --run "nixpkgs-review wip"
  • Tested execution of all binary files (usually in ./result/bin/)
  • Determined the impact on package closure size (by running nix path-info -S before and after)
  • Ensured that relevant documentation is up to date
  • Fits CONTRIBUTING.md.

@flokli
flokli requested review from Frostman, Mic92 and adisbladis March 29, 2020 13:50
@flokli

flokli commented Mar 29, 2020

Copy link
Copy Markdown
Member Author

This should be backported to 20.03 as well.

@flokli

flokli commented Mar 29, 2020

Copy link
Copy Markdown
Member Author

@GrahamcOfBorg build buildkite-agent terraform-full

@Mic92

Mic92 commented Mar 29, 2020

Copy link
Copy Markdown
Member

If you merge this to master we can as well cherry-pick ff2ea91

@ofborg ofborg Bot added the 6.topic: golang Go is a high-level general purpose programming language that is statically typed and compiled. label Mar 29, 2020
@Mic92 Mic92 mentioned this pull request Mar 29, 2020
10 tasks
@ofborg
ofborg Bot requested review from cstrahan and kalbasit March 29, 2020 14:06
@ofborg ofborg Bot added 10.rebuild-darwin: 501+ This PR causes many rebuilds on Darwin and should normally target the staging branches. 10.rebuild-darwin: 501-1000 This PR causes many rebuilds on Darwin and should normally target the staging branches. 10.rebuild-linux: 501+ This PR causes many rebuilds on Linux and should normally target the staging branches. 10.rebuild-linux: 501-1000 This PR causes many rebuilds on Linux and should normally target the staging branches. labels Mar 29, 2020
@flokli

flokli commented Mar 29, 2020

Copy link
Copy Markdown
Member Author

It seems terranix is currently broken - it refers to a non-existing rev, which is why terraform-full fails to build in CI (but the other providers build, and I hope there's no major changes in that minor release).

I can cherry-pick ff2ea91 into this PR, but can also retarget to staging - your call.

Obviously, ff2ea91 shouldn't go to 20.03, though ;-)

@Mic92

Mic92 commented Apr 2, 2020

Copy link
Copy Markdown
Member

It seems terranix is currently broken - it refers to a non-existing rev, which is why terraform-full fails to build in CI (but the other providers build, and I hope there's no major changes in that minor release).

I can cherry-pick ff2ea91 into this PR, but can also retarget to staging - your call.

Obviously, ff2ea91 shouldn't go to 20.03, though ;-)

is terranix in nixpkgs?

@flokli

flokli commented Apr 2, 2020

Copy link
Copy Markdown
Member Author

No, I screwed up, and had that (broken) in my local overlay - figuring this out was quite some fun while working on #83279 :-D

@Mic92

Mic92 commented Apr 2, 2020

Copy link
Copy Markdown
Member

@GrahamcOfBorg build docker

@Mic92

Mic92 commented Apr 2, 2020

Copy link
Copy Markdown
Member

@GrahamcOfBorg build direnv

This contains at least golang/go#37807, which
currently causes some of our go binaries to just fail when booted with
older kernels.

See golang/go#37436 for context.
@flokli
flokli changed the base branch from master to staging April 2, 2020 18:07
@ofborg ofborg Bot added 8.has: documentation This PR adds or changes documentation 8.has: module (update) This PR changes an existing module in `nixos/` 10.rebuild-darwin-stdenv This PR causes stdenv to rebuild on Darwin and must target a staging branch. 10.rebuild-linux-stdenv This PR causes stdenv to rebuild on Linux and must target a staging branch. 8.has: clean-up This PR removes packages or removes other cruft 8.has: package (new) This PR adds a new package 10.rebuild-darwin: 5001+ This PR causes many rebuilds on Darwin and must target the staging branches. 10.rebuild-linux: 5001+ This PR causes many rebuilds on Linux and must target the staging branches. and removed 6.topic: GNOME GNOME desktop environment and its underlying platform 6.topic: nixos Issues or PRs affecting NixOS modules, or package usability issues specific to NixOS 6.topic: pantheon The Pantheon desktop environment 6.topic: python Python is a high-level, general-purpose programming language. 6.topic: qt/kde Object-oriented framework for GUI creation 6.topic: rust General-purpose programming language emphasizing performance, type safety, and concurrency. 6.topic: xfce The Xfce Desktop Environment 8.has: changelog This PR adds or changes release notes 8.has: documentation This PR adds or changes documentation 8.has: module (update) This PR changes an existing module in `nixos/` 10.rebuild-darwin-stdenv This PR causes stdenv to rebuild on Darwin and must target a staging branch. 10.rebuild-linux-stdenv This PR causes stdenv to rebuild on Linux and must target a staging branch. 10.rebuild-darwin: 501-1000 This PR causes many rebuilds on Darwin and should normally target the staging branches. 10.rebuild-linux: 501-1000 This PR causes many rebuilds on Linux and should normally target the staging branches. labels Apr 2, 2020
@Mic92
Mic92 merged commit d7f86b3 into NixOS:staging Apr 2, 2020
@Mic92

Mic92 commented Apr 2, 2020

Copy link
Copy Markdown
Member

staging backport: 3b38319

@flokli
flokli deleted the go-1.14.1 branch April 2, 2020 20:01
@flokli

flokli commented Apr 2, 2020

Copy link
Copy Markdown
Member Author

^ 3b38319 was the staging-20.03 backport.

@knl knl mentioned this pull request Apr 6, 2020
10 tasks
dhess pushed a commit to hackworthltd/hacknix that referenced this pull request Apr 8, 2020
@Frostman Frostman mentioned this pull request Apr 8, 2020
10 tasks
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

6.topic: golang Go is a high-level general purpose programming language that is statically typed and compiled. 8.has: clean-up This PR removes packages or removes other cruft 8.has: package (new) This PR adds a new package 10.rebuild-darwin: 501+ This PR causes many rebuilds on Darwin and should normally target the staging branches. 10.rebuild-darwin: 5001+ This PR causes many rebuilds on Darwin and must target the staging branches. 10.rebuild-linux: 501+ This PR causes many rebuilds on Linux and should normally target the staging branches. 10.rebuild-linux: 5001+ This PR causes many rebuilds on Linux and must target the staging branches.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants