Skip to content

chore: add przemo as codeowner - #38

Merged
Orkuncakilkaya merged 1 commit into
mainfrom
add-codeowner
Mar 31, 2026
Merged

chore: add przemo as codeowner#38
Orkuncakilkaya merged 1 commit into
mainfrom
add-codeowner

Conversation

@Orkuncakilkaya

Copy link
Copy Markdown
Contributor

This pull request makes a small update to the .github/CODEOWNERS file by adding @TheUnderScorer as a code owner for the repository.

@github-actions

Copy link
Copy Markdown
Contributor

⚠️ This PR doesn't contain any changesets. If there are user-facing changes, don't forget to run:

pnpm exec changeset

to create a changeset.

@Orkuncakilkaya
Orkuncakilkaya merged commit 8256d2a into main Mar 31, 2026
8 checks passed
@Orkuncakilkaya
Orkuncakilkaya deleted the add-codeowner branch March 31, 2026 14:07
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants