Add v1.38 Release Comms Shadows to teams.yaml - #6644
RinkiyaKeDad wants to merge 1 commit into
Conversation
Signed-off-by: Arsh Sharma <arshsharma461@gmail.com>
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: RinkiyaKeDad The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
@RinkiyaKeDad: The following test failed, say
Full PR test history. Your PR dashboard. Please help us cut down on flakes by linking to an open issue when you hit one in your PR. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
|
/lgtm @RinkiyaKeDad CI is failing some some members are not org members. You can do |
*Added Comms shadows under the release-team, release-comms and milestone-maintainers.
Ref: kubernetes/sig-release#3101